Update functions-operator-bundle to 4de65c5 - #299
Conversation
5c75048 to
a743859
Compare
|
[APPROVALNOTIFIER] This PR is APPROVED Approval requirements bypassed by manually added approval. This pull-request has been approved by: red-hat-konflux-kflux-prd-rh02[bot] The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
/test e2e-install-aws |
Image created from 'https://github.com/openshift/functions-operator?rev=1ef1c3925978c0f76a033bb996177d439380cbb7' Signed-off-by: red-hat-konflux-kflux-prd-rh02 <190377777+red-hat-konflux-kflux-prd-rh02[bot]@users.noreply.github.com>
a743859 to
1d5c88b
Compare
|
@red-hat-konflux-kflux-prd-rh02[bot]: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/lgtm |
Image created from 'https://github.com/openshift/functions-operator?rev=1ef1c3925978c0f76a033bb996177d439380cbb7'
This PR contains the following updates:
26d27b5->4de65c5Configuration
📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.
To execute skipped test pipelines write comment
/ok-to-testThis is Konflux nudging PR, not Mintmaker.