Skip to content

OSDOCS#22073: Docs for OSSO 1.4.3 - #119591

Merged
bergerhoffer merged 1 commit into
openshift:enterprise-4.19from
bergerhoffer:OSDOCS-22073
Sep 9, 2026
Merged

OSDOCS#22073: Docs for OSSO 1.4.3#119591
bergerhoffer merged 1 commit into
openshift:enterprise-4.19from
bergerhoffer:OSDOCS-22073

Conversation

@bergerhoffer

@bergerhoffer bergerhoffer commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

@bergerhoffer
bergerhoffer changed the base branch from main to enterprise-4.19 September 9, 2026 13:27
@openshift-ci openshift-ci Bot added the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Sep 9, 2026
@bergerhoffer bergerhoffer changed the title Osdocs 22073 OSDOCS#22073: Docs for OSSO 1.4.3 Sep 9, 2026
@openshift-ci openshift-ci Bot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Sep 9, 2026
@bergerhoffer bergerhoffer added this to the Continuous Release milestone Sep 9, 2026
@ocpdocs-previewbot

Copy link
Copy Markdown

@bergerhoffer

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-ci

openshift-ci Bot commented Sep 9, 2026

Copy link
Copy Markdown

@bergerhoffer: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/validate-asciidoc ae77981 link true /test validate-asciidoc
ci/prow/validate-portal ae77981 link true /test validate-portal

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@ingvagabund

Copy link
Copy Markdown
Member

LGTM

@bergerhoffer
bergerhoffer merged commit a4a81db into openshift:enterprise-4.19 Sep 9, 2026
2 checks passed
@bergerhoffer

Copy link
Copy Markdown
Contributor Author

/cherrypick enterprise-4.18

@openshift-cherrypick-robot

Copy link
Copy Markdown

@bergerhoffer: new pull request created: #119629

Details

In response to this:

/cherrypick enterprise-4.18

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

branch/enterprise-4.18 branch/enterprise-4.19 size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants