fix(deps): update module github.com/gin-gonic/gin to v1.10.1 - #1175
renovate[bot] wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
2ae4e2d to
3a4f1db
Compare
3a4f1db to
9d22b9a
Compare
21129ab to
509b775
Compare
509b775 to
a980943
Compare
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: Organization UI Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
a980943 to
845da7c
Compare
This PR contains the following updates:
v1.10.0→v1.10.1Release Notes
gin-gonic/gin (github.com/gin-gonic/gin)
v1.10.1Compare Source
Features
Enhancements
Build process updates
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.