chore: Delete now unused argocd-operator manifests and the make goals using/producing them - #1297
olivergondza wants to merge 1 commit into
Conversation
… using/producing them Signed-off-by: Oliver Gondža <ogondza@gmail.com>
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Important Review skippedToo many files! This PR contains 470 files, which is 170 over the limit of 300. To get a review, reduce the PR to 300 files or fewer by splitting it into smaller PRs or changing its base branch. ⚙️ Run configurationConfiguration used: Repository YAML (base), Organization UI (inherited) Review profile: CHILL Plan: Enterprise Run ID: ⛔ Files ignored due to path filters (2)
📒 Files selected for processing (470)
You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
| @@ -1,4 +1,4 @@ | |||
| package v1beta1 | |||
| package v1alpha1 | |||
There was a problem hiding this comment.
This is the only unittest that is accessing the manifest file in FS. Moved from argocd-operator/ to the top level so it verifies those manifests now.
|
@anandf, PTAL. We discussed this one |
|
/retest |
|
@olivergondza: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
What type of PR is this?
What does this PR do / why we need it:
Have you updated the necessary documentation?
Which issue(s) this PR fixes:
Fixes #?
Test acceptance criteria:
How to test changes / Special notes to the reviewer: