Skip to content

chore(deps): bump grpc-go to 1.82.2 - #3861

Open
vladbologa wants to merge 1 commit into
release-2.39from
vb/2.39-grpc-bump
Open

vladbologa wants to merge 1 commit into
release-2.39from
vb/2.39-grpc-bump

Conversation

@vladbologa

@vladbologa vladbologa commented Sep 18, 2026

Copy link
Copy Markdown
Contributor

Fixes ROX-37059 and ROX-37060 - CVE-2026-84445.

@vladbologa
vladbologa requested a review from a team as a code owner September 18, 2026 09:05
@vladbologa

Copy link
Copy Markdown
Contributor Author

/test slim-e2e-tests

@openshift-ci

openshift-ci Bot commented Sep 18, 2026

Copy link
Copy Markdown

@vladbologa: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-tests 613aa4b link false /test e2e-tests
ci/prow/slim-e2e-tests 613aa4b link false /test slim-e2e-tests

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@vladbologa

Copy link
Copy Markdown
Contributor Author

Both e2e tests are failing on auth issues on registry.redhat.io/openshift3/logging-elasticsearch

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants